FazBrowse GitHub Viewer | Trending |
URL:
| Home
Tools: [Download Repo ZIP]   [Original HTTPS Page]

chore(deps): update dependency ipfs-core to v0.17.0 by renovate[bot] · Pull Request #103 · dstack-js/dstack · GitHub

This repository was archived by the owner on Nov 23, 2022. It is now read-only.
/ dstack Public archive

chore(deps): update dependency ipfs-core to v0.17.0 - #103

Open
renovate[bot] wants to merge 1 commit into
mainfrom
renovate/ipfs-core-0.x
Open

renovate[bot] wants to merge 1 commit into
mainfrom
renovate/ipfs-core-0.x

Conversation

renovate Bot commented May 14, 2022 •
edited
Loading

Copy link
Copy Markdown
Contributor

This PR contains the following updates:

Package Change Age Adoption Passing Confidence
ipfs-core (source) 0.14.1 -> 0.17.0

Release Notes

ipfs/js-ipfs

v0.17.0: ipfs-core ipfs-core-v0.17.0

Compare Source

⚠ BREAKING CHANGES
  • ipfs is now bundled with libp2p@0.40.x which has different config
  • require IPNS V2 signatures (#​4207)
Features
Bug Fixes
Dependencies
  • The following workspace dependencies were updated
    • dependencies
      • ipfs-core-config bumped from ^0.5.1 to ^0.6.0
      • ipfs-core-types bumped from ^0.12.1 to ^0.13.0
      • ipfs-core-utils bumped from ^0.16.1 to ^0.17.0
      • ipfs-http-client bumped from ^58.0.1 to ^59.0.0
    • devDependencies
      • interface-ipfs-core bumped from ^0.156.1 to ^0.157.0

v0.16.1: ipfs-core ipfs-core-v0.16.1

Compare Source

Bug Fixes
Dependencies
  • The following workspace dependencies were updated
    • dependencies
      • ipfs-core-config bumped from ^0.5.0 to ^0.5.1
      • ipfs-core-types bumped from ^0.12.0 to ^0.12.1
      • ipfs-core-utils bumped from ^0.16.0 to ^0.16.1
      • ipfs-http-client bumped from ^58.0.0 to ^58.0.1
    • devDependencies
      • interface-ipfs-core bumped from ^0.156.0 to ^0.156.1

v0.16.0: ipfs-core ipfs-core-v0.16.0

Compare Source

⚠ BREAKING CHANGES
deps
Dependencies
  • The following workspace dependencies were updated
    • dependencies
      • ipfs-core-config bumped from ^0.4.0 to ^0.5.0
      • ipfs-core-types bumped from ^0.11.0 to ^0.12.0
      • ipfs-core-utils bumped from ^0.15.0 to ^0.16.0
      • ipfs-http-client bumped from ^57.0.0 to ^58.0.0
    • devDependencies
      • interface-ipfs-core bumped from ^0.155.0 to ^0.156.0

v0.15.4: ipfs-core ipfs-core-v0.15.4

Compare Source

Bug Fixes
Dependencies
  • The following workspace dependencies were updated
    • dependencies
      • ipfs-http-client bumped from ^57.0.2 to ^57.0.3
    • devDependencies
      • interface-ipfs-core bumped from ^0.155.1 to ^0.155.2

v0.15.3: ipfs-core ipfs-core-v0.15.3

Compare Source

Bug Fixes
Dependencies
  • The following workspace dependencies were updated
    • dependencies
      • ipfs-core-config bumped from ^0.4.0 to ^0.4.1
      • ipfs-core-types bumped from ^0.11.0 to ^0.11.1
      • ipfs-core-utils bumped from ^0.15.0 to ^0.15.1
      • ipfs-http-client bumped from ^57.0.1 to ^57.0.2
    • devDependencies
      • interface-ipfs-core bumped from ^0.155.0 to ^0.155.1

v0.15.2: ipfs-core ipfs-core-v0.15.2

Compare Source

Bug Fixes

v0.15.1: ipfs-core ipfs-core-v0.15.1

Compare Source

Dependencies
  • The following workspace dependencies were updated
    • dependencies
      • ipfs-http-client bumped from ^57.0.0 to ^57.0.1

v0.15.0: ipfs-core ipfs-core-v0.15.0

Compare Source

⚠ BREAKING CHANGES
  • This module is now ESM only and there return types of some methods have changed
Features
Bug Fixes
Dependencies
  • The following workspace dependencies were updated
    • dependencies
      • ipfs-core-config bumped from ^0.3.2 to ^0.4.0
      • ipfs-core-types bumped from ^0.10.3 to ^0.11.0
      • ipfs-core-utils bumped from ^0.14.3 to ^0.15.0
      • ipfs-http-client bumped from ^56.0.3 to ^57.0.0
    • devDependencies
      • interface-ipfs-core bumped from ^0.154.3 to ^0.155.0

v0.14.3: ipfs-core ipfs-core-v0.14.3

Compare Source

Bug Fixes
Dependencies
  • The following workspace dependencies were updated
    • dependencies
      • ipfs-core-config bumped from ^0.3.2 to ^0.3.3
      • ipfs-core-types bumped from ^0.10.2 to ^0.10.3
      • ipfs-core-utils bumped from ^0.14.2 to ^0.14.3
      • ipfs-http-client bumped from ^56.0.2 to ^56.0.3
    • devDependencies
      • interface-ipfs-core bumped from ^0.154.2 to ^0.154.3

v0.14.2: ipfs-core ipfs-core-v0.14.2

Compare Source

Dependencies
  • The following workspace dependencies were updated
    • dependencies
      • ipfs-core-config bumped from ^0.3.1 to ^0.3.2
      • ipfs-core-types bumped from ^0.10.1 to ^0.10.2
      • ipfs-core-utils bumped from ^0.14.1 to ^0.14.2
      • ipfs-http-client bumped from ^56.0.1 to ^56.0.2
    • devDependencies
      • interface-ipfs-core bumped from ^0.154.1 to ^0.154.2

Configuration

📅 Schedule: Branch creation - At any time (no schedule defined), Automerge - At any time (no schedule defined).

🚦 Automerge: Disabled by config. Please merge this manually once you are satisfied.

♻ Rebasing: Whenever PR becomes conflicted, or you tick the rebase/retry checkbox.

🔕 Ignore: Close this PR and you won't be reminded about these updates again.


  • If you want to rebase/retry this PR, check this box

This PR has been generated by Mend Renovate. View repository job log here.

nx-cloud Bot commented May 14, 2022 •
edited
Loading

Copy link
Copy Markdown

☁️ Nx Cloud Report

CI is running/has finished running commands for commit e4372d9. As they complete they will appear below. Click to see the status, the terminal output, and the build insights.

📂 See all runs for this branch


🟥 Failed Commands
nx affected --target=build --parallel --max-parallel=3
nx build relay

Sent with 💌 from NxCloud.

netlify Bot commented May 14, 2022 •
edited
Loading

Copy link
Copy Markdown

❌ Deploy Preview for dstack-explorer failed.

Name Link
🔨 Latest commit e4372d9
🔍 Latest deploy log https://app.netlify.com/sites/dstack-explorer/deploys/635807afc1f57e0008208713

renovate Bot force-pushed the renovate/ipfs-core-0.x branch from c90baed to da0a3e7 Compare May 30, 2022 21:05
renovate Bot changed the title chore(deps): update dependency ipfs-core to v0.14.3 chore(deps): update dependency ipfs-core to v0.15.0 May 30, 2022
renovate Bot changed the title chore(deps): update dependency ipfs-core to v0.15.0 chore(deps): update dependency ipfs-core to v0.15.1 Jun 1, 2022
renovate Bot force-pushed the renovate/ipfs-core-0.x branch 2 times, most recently from 9d585b6 to 668e1dc Compare June 7, 2022 15:53
renovate Bot changed the title chore(deps): update dependency ipfs-core to v0.15.1 chore(deps): update dependency ipfs-core to v0.15.2 Jun 14, 2022
renovate Bot force-pushed the renovate/ipfs-core-0.x branch from 668e1dc to 74910c8 Compare June 14, 2022 13:07
renovate Bot changed the title chore(deps): update dependency ipfs-core to v0.15.2 chore(deps): update dependency ipfs-core to v0.15.3 Jun 23, 2022
renovate Bot force-pushed the renovate/ipfs-core-0.x branch 2 times, most recently from ee9eb60 to aa7174a Compare June 24, 2022 11:38
renovate Bot changed the title chore(deps): update dependency ipfs-core to v0.15.3 chore(deps): update dependency ipfs-core to v0.15.4 Jun 24, 2022
renovate Bot force-pushed the renovate/ipfs-core-0.x branch from aa7174a to 28d797c Compare September 7, 2022 15:37
renovate Bot changed the title chore(deps): update dependency ipfs-core to v0.15.4 chore(deps): update dependency ipfs-core to v0.16.0 Sep 7, 2022
renovate Bot force-pushed the renovate/ipfs-core-0.x branch from 28d797c to cd67f52 Compare September 22, 2022 00:44
renovate Bot changed the title chore(deps): update dependency ipfs-core to v0.16.0 chore(deps): update dependency ipfs-core to v0.16.1 Sep 22, 2022
renovate Bot force-pushed the renovate/ipfs-core-0.x branch from cd67f52 to e4372d9 Compare October 25, 2022 15:58
renovate Bot changed the title chore(deps): update dependency ipfs-core to v0.16.1 chore(deps): update dependency ipfs-core to v0.17.0 Oct 25, 2022
This file contains hidden or bidirectional Unicode text that may be interpreted or compiled differently than what appears below. To review, open the file in an editor that reveals hidden Unicode characters. Learn more about bidirectional Unicode characters
Sign up for free to subscribe to this conversation on GitHub. Already have an account? Sign in.

Labels

None yet

Projects

None yet

Development

Successfully merging this pull request may close these issues.

0 participants


Back | FazBrowse Home | New Git URL