FazBrowse GitHub Viewer | Trending |
URL:
| Home
Tools: [Download Repo ZIP]   [Original HTTPS Page]

fix(deps): update dependency abortable-iterator to v4 by renovate[bot] · Pull Request #85 · dstack-js/dstack · GitHub

This repository was archived by the owner on Nov 23, 2022. It is now read-only.
/ dstack Public archive

fix(deps): update dependency abortable-iterator to v4 - #85

Closed
renovate[bot] wants to merge 1 commit into
mainfrom
renovate/abortable-iterator-4.x
Closed

renovate[bot] wants to merge 1 commit into
mainfrom
renovate/abortable-iterator-4.x

Conversation

renovate Bot commented Mar 25, 2022

Copy link
Copy Markdown
Contributor

This PR contains the following updates:

Package Change Age Adoption Passing Confidence
abortable-iterator ^3.0.0 -> 4.0.2

Release Notes

alanshaw/abortable-iterator

v4.0.2

Compare Source

v4.0.1

Compare Source

v4.0.0

Compare Source

⚠ BREAKING CHANGES
  • switch to named exports, ESM only
Features

Configuration

📅 Schedule: At any time (no schedule defined).

🚦 Automerge: Disabled by config. Please merge this manually once you are satisfied.

♻ Rebasing: Whenever PR becomes conflicted, or you tick the rebase/retry checkbox.

🔕 Ignore: Close this PR and you won't be reminded about this update again.


  • If you want to rebase/retry this PR, click this checkbox.

This PR has been generated by WhiteSource Renovate. View repository job log here.

nx-cloud Bot commented Mar 25, 2022 •
edited
Loading

Copy link
Copy Markdown

☁️ Nx Cloud Report

CI is running/has finished running commands for commit 2713b22. As they complete they will appear below. Click to see the status, the terminal output, and the build insights.

📂 See all runs for this branch


🟥 Failed Commands
nx affected --target=build --parallel --max-parallel=3
✅ Successfully ran 1 target

Sent with 💌 from NxCloud.

netlify Bot commented Mar 25, 2022 •
edited
Loading

Copy link
Copy Markdown

❌ Deploy Preview for dstack-explorer failed.

Name Link
🔨 Latest commit 2713b22
🔍 Latest deploy log https://app.netlify.com/sites/dstack-explorer/deploys/623da457bd9ce30009b72f64

0x77dev closed this Mar 28, 2022
0x77dev deleted the renovate/abortable-iterator-4.x branch March 28, 2022 14:32

renovate Bot commented Mar 28, 2022

Copy link
Copy Markdown
Contributor Author

Renovate Ignore Notification

As this PR has been closed unmerged, Renovate will ignore this upgrade and you will not receive PRs for any future 4.x releases. However, if you upgrade to 4.x manually then Renovate will reenable minor and patch updates automatically.

If this PR was closed by mistake or you changed your mind, you can simply rename this PR and you will soon get a fresh replacement PR opened.

This file contains hidden or bidirectional Unicode text that may be interpreted or compiled differently than what appears below. To review, open the file in an editor that reveals hidden Unicode characters. Learn more about bidirectional Unicode characters
Sign up for free to subscribe to this conversation on GitHub. Already have an account? Sign in.

Labels

None yet

Projects

None yet

Development

Successfully merging this pull request may close these issues.

2 participants


Back | FazBrowse Home | New Git URL