| FazBrowse GitHub Viewer | Trending | | Home |
| Tools: [Download Repo ZIP] [Original HTTPS Page] |
Sorry, something went wrong.
…t` instead (#112161) * Change workflows to use pull_request_target instead of pull_request event * Add CODEOWNERS entry * Add initial readme
* Condition labeling workflows to only run on dotnet/runtime. * Improve readme * Add jeffhandley as explicit workflow owner Co-authored-by: Jeff Handley <jeffhandley@users.noreply.github.com>
|
Tagging subscribers to this area: @dotnet/runtime-infrastructure |
Sorry, something went wrong.
| Back | FazBrowse Home | New Git URL |
cc @jeffhandley
Backporting #112169 and #112161
Infra only, tell-mode.
Backporting to the release/9.0 branch so it can flow to release/9.0-staging automatically at any time.
I discussed the functionality with @jeffhandley via chat, and we should not expect the pull_request_target change to work in this PR at the moment. It will start working after this is merged.