| FazBrowse GitHub Viewer | Trending | | Home |
| Tools: [Download Repo ZIP] [Original HTTPS Page] |
Sorry, something went wrong.
No need to run this workflow on all PRs, only those that change the Python dependency installation mechanism.
Mergeback v2.1.17 refs/heads/releases/v2 into main
…-path-filters CI: Add path filters to python-deps workflow
Reduce the number of concurrent jobs. This will require a branch protection rule update, renaming `check-js` to Check JS` and removing `Lint`.
…t-check-js CI: Combine JS lint and JS check jobs
Refactor the PR checks workflow into two workflows: PR checks and Runner checks. This does not change the actual check jobs that are run. It also does not change the expected check names (which rely only on the job name, not the workflow name). This makes it easier to inspect workflow run summaries in the UI and to separately retry subsets of failed jobs in case of flakiness. In future we will clean up the Runner checks, since this is a deprecated component.
…nner-checks CI: Move Runner checks into their own workflow
Reduce the number of concurrent jobs by 1. Run these checks in succession instead, as the `check-file-contents` job in the PR checks workflow.
…eck-file-contents CI: Combine verify-pr-checks and check-for-conflicts jobs
Update default CodeQL version to 2.10.2
Merge main into releases/v2
This reverts commit d8ed0a3.
This reverts commit 624285f.
| Back | FazBrowse Home | New Git URL |
Merging 2ca79b6 into releases/v1
Conductor for this PR is @cklin
Contains the following pull requests:
Please review the following: