FazBrowse GitHub Viewer | Trending |
URL:
| Home
Tools: [Download Repo ZIP]   [Original HTTPS Page]

fix misspelling functionallity -> functionality by Abirdcfly · Pull Request #78 · knative/hack · GitHub

/ hack Public

fix misspelling functionallity -> functionality - #78

Merged
knative-prow-robot merged 1 commit into
knative:mainfrom
Abirdcfly:main
Oct 15, 2021
Merged

knative-prow-robot merged 1 commit into
knative:mainfrom
Abirdcfly:main

Conversation

Copy link
Copy Markdown
Contributor

Changes

/kind

Fixes #

Release Note

Docs

Copy link
Copy Markdown

@Abirdcfly: The label(s) kind/<kind> cannot be applied, because the repository doesn't have them.

Details

In response to this:

Changes

/kind

Fixes #

Release Note

Docs

Instructions for interacting with me using PR comments are available here. If you have questions or suggestions related to my behavior, please file an issue against the kubernetes/test-infra repository.

google-cla Bot added the cla: yes Indicates the PR's author has signed the CLA. label Sep 16, 2021

Copy link
Copy Markdown

Welcome @Abirdcfly! It looks like this is your first PR to knative/hack 🎉

Copy link
Copy Markdown

Hi @Abirdcfly. Thanks for your PR.

I'm waiting for a knative member to verify that this patch is reasonable to test. If it is, they should reply with /ok-to-test on its own line. Until that is done, I will not automatically test new commits in this PR, but the usual testing commands by org members will still work. Regular contributors should join the org to skip this step.

Once the patch is verified, the new status will be reflected by the ok-to-test label.

I understand the commands that are listed here.

Details

Instructions for interacting with me using PR comments are available here. If you have questions or suggestions related to my behavior, please file an issue against the kubernetes/test-infra repository.

knative-prow-robot added needs-ok-to-test Indicates a PR that requires an org member to verify it is safe to test. size/XS Denotes a PR that changes 0-9 lines, ignoring generated files. labels Sep 16, 2021

Copy link
Copy Markdown
Contributor

/lgtm
Thanks @Abirdcfly

knative-prow-robot added the lgtm Indicates that a PR is ready to be merged. label Oct 13, 2021

julz commented Oct 15, 2021

Copy link
Copy Markdown

/retest

chizhg commented Oct 15, 2021

Copy link
Copy Markdown
Contributor

/ok-to-test

knative-prow-robot added ok-to-test Indicates a non-member PR verified by an org member that is safe to test. and removed needs-ok-to-test Indicates a PR that requires an org member to verify it is safe to test. labels Oct 15, 2021

chizhg commented Oct 15, 2021

Copy link
Copy Markdown
Contributor

/lgtm
/approve

Copy link
Copy Markdown

[APPROVALNOTIFIER] This PR is APPROVED

This pull-request has been approved by: Abirdcfly, chizhg

The full list of commands accepted by this bot can be found here.

The pull request process is described here

Details Needs approval from an approver in each of these files:

Approvers can indicate their approval by writing /approve in a comment
Approvers can cancel approval by writing /approve cancel in a comment

knative-prow-robot added the approved Indicates a PR has been approved by an approver from all required OWNERS files. label Oct 15, 2021

chizhg commented Oct 15, 2021

Copy link
Copy Markdown
Contributor

/retest

knative-prow-robot merged commit 51582ce into knative:main Oct 15, 2021
This file contains hidden or bidirectional Unicode text that may be interpreted or compiled differently than what appears below. To review, open the file in an editor that reveals hidden Unicode characters. Learn more about bidirectional Unicode characters
Sign up for free to join this conversation on GitHub. Already have an account? Sign in to comment

Labels

approved Indicates a PR has been approved by an approver from all required OWNERS files. cla: yes Indicates the PR's author has signed the CLA. lgtm Indicates that a PR is ready to be merged. ok-to-test Indicates a non-member PR verified by an org member that is safe to test. size/XS Denotes a PR that changes 0-9 lines, ignoring generated files.

Projects

None yet

Development

Successfully merging this pull request may close these issues.

5 participants


Back | FazBrowse Home | New Git URL