FazBrowse GitHub Viewer | Trending |
URL:
| Home
Tools: [Download Repo ZIP]   [Original HTTPS Page]

fix(db): do not over allocate memory by janisz · Pull Request #8542 · stackrox/stackrox · GitHub

fix(db): do not over allocate memory - #8542

Merged
janisz merged 1 commit into
masterfrom
do_not_over_allocate_mem_in_copy_from
Nov 8, 2023
Merged

fix(db): do not over allocate memory#8542
janisz merged 1 commit into
masterfrom
do_not_over_allocate_mem_in_copy_from

Conversation

janisz commented Nov 8, 2023
edited
Loading

Copy link
Copy Markdown
Contributor

Description

This PR prevents over allocating memory when using upsert many.

Datastore_Add_and_Flush/add_1000_events_to_the_writer_and_flush_it-8  3.153Mi ± 0%     2.786Mi ± 0%  -11.62% (p=0.000 n=30+10)
Datastore_Add_and_Flush/add_5000_events_to_the_writer_and_flush_it-8  20.49Mi ± 0%     18.76Mi ± 0%   -8.41% (p=0.000 n=30+10)
Datastore_Add_and_Flush/add_10000_events_to_the_writer_and_flush_it-8 41.00Mi ± 0%     37.48Mi ± 0%   -8.57% (p=0.000 n=30+10)
Datastore_Add_and_Flush/add_20000_events_to_the_writer_and_flush_it-8 81.67Mi ± 0%     74.57Mi ± 0%   -8.70% (p=0.000 n=30+10)

Checklist

  • Investigated and inspected CI test results
  • Unit test and regression tests added
  • Evaluated and added CHANGELOG entry if required
  • Determined and documented upgrade steps
  • Documented user facing changes (create PR based on openshift/openshift-docs and merge into rhacs-docs)

If any of these don't apply, please comment below.

Testing Performed

Here I tell how I validated my change

CI

Reminder for reviewers

In addition to reviewing code here, reviewers must also review testing and request further testing in case the
performed one does not seem sufficient. As a reviewer, you must not approve the change until you understand the
performed testing and you are satisfied with it.

Signed-off-by: Tomasz Janiszewski <tomek@redhat.com>
janisz force-pushed the do_not_over_allocate_mem_in_copy_from branch from 1d2c8e6 to dbe4bc0 Compare November 8, 2023 10:03

ghost commented Nov 8, 2023

Copy link
Copy Markdown

Images are ready for the commit at dbe4bc0.

To use with deploy scripts, first export MAIN_IMAGE_TAG=4.3.x-104-gdbe4bc0ac7.

openshift-ci Bot commented Nov 8, 2023

Copy link
Copy Markdown

@janisz: The following test failed, say /retest to rerun all failed tests or /retest-required to rerun all mandatory failed tests:

Test name Commit Details Required Rerun command
ci/prow/ocp-4-11-qa-e2e-tests dbe4bc0 link false /test ocp-4-11-qa-e2e-tests

Full PR test history. Your PR dashboard.

Details

Instructions for interacting with me using PR comments are available here. If you have questions or suggestions related to my behavior, please file an issue against the kubernetes/test-infra repository. I understand the commands that are listed here.

janisz commented Nov 8, 2023

Copy link
Copy Markdown
Contributor Author

/retest

janisz merged commit 99fa7b3 into master Nov 8, 2023
janisz deleted the do_not_over_allocate_mem_in_copy_from branch November 8, 2023 17:14
ksurabhi91 pushed a commit that referenced this pull request Nov 28, 2023
Signed-off-by: Tomasz Janiszewski <tomek@redhat.com>
This file contains hidden or bidirectional Unicode text that may be interpreted or compiled differently than what appears below. To review, open the file in an editor that reveals hidden Unicode characters. Learn more about bidirectional Unicode characters
Sign up for free to join this conversation on GitHub. Already have an account? Sign in to comment

Projects

None yet

Development

Successfully merging this pull request may close these issues.

2 participants


Back | FazBrowse Home | New Git URL