FazBrowse GitHub Viewer | Trending |
URL:
| Home
Tools: [Download Repo ZIP]   [Original HTTPS Page]

:broom: Removing the vendor dir :fireworks: by cardil · Pull Request #1966 · knative/func · GitHub

/ func Public

🧹 Removing the vendor dir 🎆 - #1966

Merged
knative-prow[bot] merged 1 commit into
knative:mainfrom
cardil:cleanup/vendorless
Sep 25, 2023
Merged

knative-prow[bot] merged 1 commit into
knative:mainfrom
cardil:cleanup/vendorless

Conversation

cardil commented Sep 7, 2023

Copy link
Copy Markdown
Contributor

Changes

  • 🧹 Removing the vendor dir 🎆

/kind cleanup

Tests knative/hack#311
Tests knative/hack#222
Related to knative-extensions/kn-plugin-event#307

knative-prow Bot added kind/cleanup Cleanup size/XXL 🤖 PR changes 1000+ lines, ignoring generated files. labels Sep 7, 2023
knative-prow Bot requested review from nainaz and navidshaikh September 7, 2023 10:53
cardil mentioned this pull request Sep 7, 2023
17 tasks

codecov Bot commented Sep 7, 2023
edited
Loading

Copy link
Copy Markdown

Codecov Report

All modified lines are covered by tests ✅

Comparison is base (26614eb) 60.64% compared to head (5963269) 62.75%.
Report is 13 commits behind head on main.

Additional details and impacted files
@@            Coverage Diff             @@
##             main    #1966      +/-   ##
==========================================
+ Coverage   60.64%   62.75%   +2.11%     
==========================================
  Files         106      106              
  Lines       13693    13693              
==========================================
+ Hits         8304     8593     +289     
+ Misses       4569     4231     -338     
- Partials      820      869      +49     
Flag Coverage Δ
e2e-test 36.46% <ø> (+1.27%) ⬆️
e2e-test-oncluster 30.50% <ø> (?)
e2e-test-oncluster-runtime 26.96% <ø> (?)
e2e-test-runtime-go 25.59% <ø> (?)
e2e-test-runtime-node 26.60% <ø> (?)
e2e-test-runtime-python 26.60% <ø> (?)
e2e-test-runtime-quarkus 26.71% <ø> (?)
e2e-test-runtime-rust 25.59% <ø> (?)
e2e-test-runtime-springboot 25.73% <ø> (?)
integration-tests 51.42% <ø> (+2.09%) ⬆️
unit-tests-macos-latest 48.21% <ø> (+0.02%) ⬆️
unit-tests-ubuntu-latest 48.98% <ø> (-0.03%) ⬇️
unit-tests-windows-latest 48.25% <ø> (ø)

Flags with carried forward coverage won't be shown. Click here to find out more.

see 15 files with indirect coverage changes

☔ View full report in Codecov by Sentry.
📢 Have feedback on the report? Share it here.

cardil commented Sep 7, 2023

Copy link
Copy Markdown
Contributor Author

Error from https://github.com/knative/func/actions/runs/6108788480/job/16578494890?pr=1966#step:10:199 looks like infra related:

Error: preparing container for next step: creating build container: creating container: creating read-write layer with ID "75ac42ff146caa497d0a1b90a87bd6db05861249d055ee23e0557339b2161ec0": no space left on device

cardil commented Sep 7, 2023

Copy link
Copy Markdown
Contributor Author

https://github.com/knative/func/actions/runs/6108788480/job/16580443298#step:10:1 once again: no space left on the device

cardil commented Sep 7, 2023

Copy link
Copy Markdown
Contributor Author

This PR is looking good (apart from obvious infra failures - seen above).

However, I'll put this on hold, since I'll be on PTO next week.

/hold

knative-prow Bot added the do-not-merge/hold 🤖 PR should not merge because someone has issued a /hold command. label Sep 7, 2023

Copy link
Copy Markdown
Member

/lgtm

Thanks for the improvement! Re-running the E2E test...

knative-prow Bot added the lgtm 🤖 PR is ready to be merged. label Sep 20, 2023

Copy link
Copy Markdown
Contributor

@cardil to fix the failing test please rebase on main.

knative-prow Bot removed the lgtm 🤖 PR is ready to be merged. label Sep 25, 2023

knative-prow Bot commented Sep 25, 2023

Copy link
Copy Markdown

New changes are detected. LGTM label has been removed.

lkingland added lgtm 🤖 PR is ready to be merged. approved 🤖 PR has been approved by an approver from all required OWNERS files. and removed do-not-merge/hold 🤖 PR should not merge because someone has issued a /hold command. labels Sep 25, 2023

knative-prow Bot commented Sep 25, 2023

Copy link
Copy Markdown

[APPROVALNOTIFIER] This PR is APPROVED

Approval requirements bypassed by manually added approval.

This pull-request has been approved by: cardil

The full list of commands accepted by this bot can be found here.

The pull request process is described here

Details Needs approval from an approver in each of these files:

Approvers can indicate their approval by writing /approve in a comment
Approvers can cancel approval by writing /approve cancel in a comment

knative-prow Bot merged commit b6fb880 into knative:main Sep 25, 2023
cardil deleted the cleanup/vendorless branch September 26, 2023 11:20
This file contains hidden or bidirectional Unicode text that may be interpreted or compiled differently than what appears below. To review, open the file in an editor that reveals hidden Unicode characters. Learn more about bidirectional Unicode characters
Sign up for free to join this conversation on GitHub. Already have an account? Sign in to comment

Labels

approved 🤖 PR has been approved by an approver from all required OWNERS files. kind/cleanup Cleanup lgtm 🤖 PR is ready to be merged. size/XXL 🤖 PR changes 1000+ lines, ignoring generated files.

Projects

None yet

Development

Successfully merging this pull request may close these issues.

3 participants


Back | FazBrowse Home | New Git URL