FazBrowse GitHub Viewer | Trending |
URL:
| Home
Tools: [Download Repo ZIP]   [Original HTTPS Page]

:gift: Don't force the vendoring of deps by cardil · Pull Request #311 · knative/hack · GitHub

/ hack Public

🎁 Don't force the vendoring of deps - #311

Merged
knative-prow[bot] merged 1 commit into
knative:mainfrom
cardil:feature/vendorless-mode
Sep 6, 2023
Merged

knative-prow[bot] merged 1 commit into
knative:mainfrom
cardil:feature/vendorless-mode

Conversation

cardil commented Sep 5, 2023

Copy link
Copy Markdown
Contributor

Changes

  • 🎁 Don't force the vendoring of deps

/kind enhancement

Fixes #310
Requires #222

Release Note

The vendor directory isn't enforced in Knative projects anymore

knative-prow Bot commented Sep 5, 2023

Copy link
Copy Markdown

Skipping CI for Draft Pull Request.
If you want CI signal for your change, please convert it to an actual PR.
You can still manually trigger a test run with /test all

knative-prow Bot added do-not-merge/work-in-progress Indicates that a PR should not merge because it is a work in progress. kind/enhancement labels Sep 5, 2023
knative-prow Bot added approved Indicates a PR has been approved by an approver from all required OWNERS files. size/XXL Denotes a PR that changes 1000+ lines, ignoring generated files. labels Sep 5, 2023
cardil force-pushed the feature/vendorless-mode branch from b3fc86b to a5a1861 Compare September 6, 2023 16:22
cardil force-pushed the feature/vendorless-mode branch from a5a1861 to f79fbb4 Compare September 6, 2023 16:43
cardil changed the title [WIP] 🎁 Don't force the vendoring of deps 🎁 Don't force the vendoring of deps Sep 6, 2023
cardil marked this pull request as ready for review September 6, 2023 17:07
knative-prow Bot removed the do-not-merge/work-in-progress Indicates that a PR should not merge because it is a work in progress. label Sep 6, 2023
knative-prow Bot requested a review from upodroid September 6, 2023 17:08

cardil commented Sep 6, 2023

Copy link
Copy Markdown
Contributor Author

/cc @kvmware
/cc @upodroid
/cc @mgencur

knative-prow Bot requested a review from mgencur September 6, 2023 17:08

upodroid left a comment

Copy link
Copy Markdown
Member

Choose a reason for hiding this comment

The reason will be displayed to describe this comment to others. Learn more.

Choose a reason Spam Abuse Off Topic Outdated Duplicate Resolved Low Quality

LGTM

knative-prow Bot added the lgtm Indicates that a PR is ready to be merged. label Sep 6, 2023

knative-prow Bot commented Sep 6, 2023

Copy link
Copy Markdown

[APPROVALNOTIFIER] This PR is APPROVED

This pull-request has been approved by: cardil, upodroid

The full list of commands accepted by this bot can be found here.

The pull request process is described here

Details Needs approval from an approver in each of these files:

Approvers can indicate their approval by writing /approve in a comment
Approvers can cancel approval by writing /approve cancel in a comment

knative-prow Bot merged commit 760813a into knative:main Sep 6, 2023
cardil deleted the feature/vendorless-mode branch September 6, 2023 17:40
Comment thread go.work
.
schema
test
test/vendorproj

cardil Sep 6, 2023
edited
Loading

Copy link
Copy Markdown
Contributor Author

Choose a reason for hiding this comment

The reason will be displayed to describe this comment to others. Learn more.

Choose a reason Spam Abuse Off Topic Outdated Duplicate Resolved Low Quality

For the record: I added this submodule to test if the scripts allow having a mixed project, with some modules having vendor/ and some don't.

This file contains hidden or bidirectional Unicode text that may be interpreted or compiled differently than what appears below. To review, open the file in an editor that reveals hidden Unicode characters. Learn more about bidirectional Unicode characters
Sign up for free to join this conversation on GitHub. Already have an account? Sign in to comment

Labels

approved Indicates a PR has been approved by an approver from all required OWNERS files. kind/enhancement lgtm Indicates that a PR is ready to be merged. size/XXL Denotes a PR that changes 1000+ lines, ignoring generated files.

Projects

None yet

Development

Successfully merging this pull request may close these issues.

Hack scripts shouldn't enforce vendor dir

2 participants


Back | FazBrowse Home | New Git URL